This repository was archived by the owner on Apr 13, 2026. It is now read-only.
chore: move website to the monorepo - #799
Merged
jakobmoellerdev merged 1 commit intoApr 13, 2026
Merged
Conversation
❌ Deploy Preview for open-component-model failed. Why did it fail? →
|
|
Important Review skippedToo many files! This PR contains 235 files, which is 85 over the limit of 150. ⚙️ Run configurationConfiguration used: Organization UI Review profile: CHILL Plan: Pro Run ID: ⛔ Files ignored due to path filters (65)
📒 Files selected for processing (235)
You can disable this status message by setting the Use the checkbox below for a quick retry:
✨ Finishing Touches🧪 Generate unit tests (beta)
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
jakobmoellerdev
previously approved these changes
Apr 13, 2026
On-behalf-of: Gergely Brautigam <gergely.brautigam@sap.com> Signed-off-by: Gergely Brautigam <182850+Skarlso@users.noreply.github.com>
jakobmoellerdev
approved these changes
Apr 13, 2026
jakobmoellerdev
merged commit Apr 13, 2026
75dce89
into
open-component-model:main
3 of 7 checks passed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to subscribe to this conversation on GitHub.
Already have an account?
Sign in.
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
What this PR does / why we need it
Archive notice because we moved the repo. https://github.com/open-component-model/open-component-model/tree/main/website
Which issue(s) this PR is related to
Type of content
getting-started/ortutorials/)how-to/)concepts/)reference/)Checklist