nix-prefetch: fix prefetching when using the hash key - #190735
Merged
Conversation
Member
|
Thanks, works for me as well. |
Contributor
|
Member
Author
|
@zowoq Can you try with msteen/nix-prefetch@fbe879e applied instead? This fixes prefetching that expression, but I want to make sure I didn't break anything else in the process. (I believe) this fixes the original issue properly, as opposed to the original patch that just happened to work in the cases hexa and I tested. Thanks! |
Contributor
|
Previous one is fixed but found another: |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Description of changes
Backporting msteen/nix-prefetch#43 since it seems that the author/maintainer is MIA, and this is somewhat important.
Things done
sandbox = trueset innix.conf? (See Nix manual)nix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)nixos/doc/manual/md-to-db.shto update generated release notes